Skip to content

[release-1.9] add release notes for v1.9.2 - #10129

Merged
kkk777-7 merged 2 commits into
envoyproxy:mainfrom
kkk777-7:add-release-notes-v1.9.2
Sep 28, 2026
Merged

kkk777-7 merged 2 commits into
envoyproxy:mainfrom
kkk777-7:add-release-notes-v1.9.2

Conversation

@kkk777-7

Copy link
Copy Markdown
Member

Added release notes for v1.9.2

@kkk777-7
kkk777-7 requested a review from a team as a code owner September 28, 2026 04:56
@netlify

netlify Bot commented Sep 28, 2026 •

Copy link
Copy Markdown

✅ Deploy Preview for cerulean-figolla-1f9435 ready!

Name Link
🔨 Latest commit b017d39
🔍 Latest deploy log https://app.netlify.com/projects/cerulean-figolla-1f9435/deploys/6aba1abb3466ee0008193401
😎 Deploy Preview https://deploy-preview-10129--cerulean-figolla-1f9435.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@codecov

codecov Bot commented Sep 28, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 81.43%. Comparing base (0051ca9) to head (b017d39).

Additional details and impacted files
@@           Coverage Diff           @@
##             main   #10129   +/-   ##
=======================================
  Coverage   81.43%   81.43%           
=======================================
  Files         266      266           
  Lines       41300    41300           
=======================================
+ Hits        33632    33633    +1     
+ Misses       7667     7666    -1     
  Partials        1        1           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

Signed-off-by: kkk777-7 <kota.kimura0725@gmail.com>
Signed-off-by: kkk777-7 <kota.kimura0725@gmail.com>
@kkk777-7

Copy link
Copy Markdown
Member Author

/retest

@kkk777-7
kkk777-7 merged commit 0049470 into envoyproxy:main Sep 28, 2026
74 of 78 checks passed
kkk777-7 added a commit that referenced this pull request Sep 28, 2026
[release-1.9] add release notes for v1.9.2 (#10129)

* add: release notes for v1.9.2



* update release notes



---------

Signed-off-by: kkk777-7 <kota.kimura0725@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants