Skip to content

chore(deps): bump docker/setup-buildx-action from 3.12.0 to 4.3.0 - #449

Merged
poiana merged 1 commit into
masterfrom
dependabot/github_actions/docker/setup-buildx-action-4.3.0
Sep 2, 2026
Merged

poiana merged 1 commit into
masterfrom
dependabot/github_actions/docker/setup-buildx-action-4.3.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Aug 24, 2026

Copy link
Copy Markdown
Contributor

Bumps docker/setup-buildx-action from 3.12.0 to 4.3.0.

Release notes

Sourced from docker/setup-buildx-action's releases.

v4.3.0

Full Changelog: docker/setup-buildx-action@v4.2.0...v4.3.0

v4.2.0

Full Changelog: docker/setup-buildx-action@v4.1.0...v4.2.0

v4.1.0

Full Changelog: docker/setup-buildx-action@v4.0.0...v4.1.0

v4.0.0

... (truncated)

Commits
  • 37fe631 Merge pull request #595 from docker/dependabot/npm_and_yarn/docker/actions-to...
  • b5c4f91 [dependabot skip] chore: update generated content
  • 3e93b63 build(deps): bump @​docker/actions-toolkit from 0.92.0 to 0.95.0
  • e527031 Merge pull request #600 from docker/dependabot/npm_and_yarn/brace-expansion-1...
  • c68814b [dependabot skip] chore: update generated content
  • 3f891b0 build(deps): bump brace-expansion from 1.1.13 to 1.1.18
  • 787db26 Merge pull request #585 from docker/dependabot/npm_and_yarn/js-yaml-5.2.1
  • f779368 [dependabot skip] chore: update generated content
  • 7d5e604 build(deps): bump js-yaml from 5.2.0 to 5.3.0
  • 292c2fb Merge pull request #590 from docker/dependabot/github_actions/actions/setup-n...
  • Additional commits viewable in compare view

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code labels Aug 24, 2026
@poiana
poiana requested review from LucaGuerra and leodido August 24, 2026 15:05
@poiana

poiana commented Aug 24, 2026

Copy link
Copy Markdown

There is not a label identifying the kind of this PR.
Please specify it either using /kind <group> or manually from the side menu.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@poiana poiana added the size/XS label Aug 24, 2026
@dependabot
dependabot Bot force-pushed the dependabot/github_actions/docker/setup-buildx-action-4.3.0 branch from cd17821 to 845465d Compare September 1, 2026 15:19
Bumps [docker/setup-buildx-action](https://github.com/docker/setup-buildx-action) from 3.12.0 to 4.3.0.
- [Release notes](https://github.com/docker/setup-buildx-action/releases)
- [Commits](docker/setup-buildx-action@8d2750c...37fe631)

---
updated-dependencies:
- dependency-name: docker/setup-buildx-action
  dependency-version: 4.3.0
  dependency-type: direct:production
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/github_actions/docker/setup-buildx-action-4.3.0 branch from 845465d to 6e06fff Compare September 1, 2026 15:28
@poiana poiana added the lgtm label Sep 2, 2026
@poiana

poiana commented Sep 2, 2026

Copy link
Copy Markdown

LGTM label has been added.

DetailsGit tree hash: 0daf45918d8e5d552901d17e78c9f62f9454d237

@poiana

poiana commented Sep 2, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dependabot[bot], leogr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@poiana poiana added the approved label Sep 2, 2026
@poiana
poiana merged commit 4d02525 into master Sep 2, 2026
7 checks passed
@poiana
poiana deleted the dependabot/github_actions/docker/setup-buildx-action-4.3.0 branch September 2, 2026 07:05
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved dco-signoff: yes dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code lgtm needs-kind size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants