Skip to content

Add pmeida as a knative & knative-extensions member - #1794

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
pmeida:add-pmeida-member
Sep 23, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
pmeida:add-pmeida-member

Conversation

@pmeida

@pmeida pmeida commented Sep 11, 2026 •

Copy link
Copy Markdown
Contributor

Adding myself as a knative and knative-extensions member so I don't have to wait for the ok-to-test labels.

Contributions so far:

@knative-prow
knative-prow Bot requested review from arsenetar and matzew September 11, 2026 13:51
@knative-prow

knative-prow Bot commented Sep 11, 2026

Copy link
Copy Markdown

Welcome @pmeida! It looks like this is your first PR to knative/community 🎉

@knative-prow knative-prow Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Sep 11, 2026
@knative-prow

knative-prow Bot commented Sep 11, 2026

Copy link
Copy Markdown

Hi @pmeida. Thanks for your PR.

I'm waiting for a knative member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 11, 2026
@dsimansk

Copy link
Copy Markdown
Contributor

/ok-to-test

@knative-prow knative-prow Bot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Sep 16, 2026
@dsimansk

Copy link
Copy Markdown
Contributor

PTAL
@knative/steering-committee

@matzew matzew left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 23, 2026
@knative-prow

knative-prow Bot commented Sep 23, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: matzew, pmeida

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 23, 2026
@knative-prow
knative-prow Bot merged commit ad9de24 into knative:main Sep 23, 2026
6 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants