Skip to content

[dde-network-core] Updates for project Deepin Desktop Environment - #630

Open
transifex-integration[bot] wants to merge 4 commits into
masterfrom
transifex_update_78bc70df5b
Open

transifex-integration[bot] wants to merge 4 commits into
masterfrom
transifex_update_78bc70df5b

Conversation

@transifex-integration

@transifex-integration transifex-integration Bot commented Sep 14, 2026 •

Copy link
Copy Markdown
Contributor

The following localization files have been updated:

Details

Parameter Value
Transifex Project Deepin Desktop Environment
Grouping All Resources and Languages

Files

Source File Translation File Resource Language Type Mode
dcc-network/translations/network_en_US.ts dcc-network/translations/network_ca.ts dcc-network/translations/network_en_US.ts (master) ca TRANSLATION default
dcc-network/translations/network_en_US.ts dcc-network/translations/network_fi.ts dcc-network/translations/network_en_US.ts (master) fi TRANSLATION default
dcc-network/translations/network_en_US.ts dcc-network/translations/network_pt_BR.ts dcc-network/translations/network_en_US.ts (master) pt_BR TRANSLATION default

100% translated source file: 'network_en_US.ts'
on 'ca'.
97% of minimum 50% translated source file: 'network_en_US.ts'
on 'fi'.

Sync of partially translated files: 
untranslated content is included with an empty translation 
or source language content depending on file format
100% translated source file: 'network_en_US.ts'
on 'fi'.
100% translated source file: 'network_en_US.ts'
on 'pt_BR'.
@deepin-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: transifex-integration[bot]

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@deepin-bot

deepin-bot Bot commented Sep 22, 2026

Copy link
Copy Markdown
Contributor

TAG Bot

New tag: 2.0.102
DISTRIBUTION: unstable
Suggest: synchronizing this PR through rebase #637

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant