Skip to content

NE-2819: Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 9d977d0 - #553

Open
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
release-1.2from
konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8
Open

NE-2819: Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 9d977d0#553
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
release-1.2from
konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8

Conversation

@red-hat-konflux-kflux-prd-rh03

@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot commented Aug 6, 2026

Copy link
Copy Markdown
Contributor

Image created from 'https://github.com/openshift/external-dns-operator?rev=0ce57338b52aae45afdf19e92096d0278601c7b7'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/external-dns-operator-tenant/ext-dns-optr-1-2-rhel-8/external-dns-operator-container-ext-dns-optr-1-2-rhel-8 digest 7c60e6a -> 9d977d0

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

@openshift-ci
openshift-ci Bot requested review from Thealisyed and gcs278 August 6, 2026 14:52
@openshift-ci openshift-ci Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Aug 6, 2026
@openshift-ci

openshift-ci Bot commented Aug 6, 2026

Copy link
Copy Markdown
Contributor

Hi @red-hat-konflux-kflux-prd-rh03[bot]. Thanks for your PR.

I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@gcs278

gcs278 commented Aug 12, 2026

Copy link
Copy Markdown
Contributor

/assign @grzpiotrowski

@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 6cd4994 Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 0ce5539 Aug 27, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8 branch from 94366f3 to f24dbe0 Compare August 27, 2026 09:08
@openshift-ci

openshift-ci Bot commented Aug 27, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please ask for approval from grzpiotrowski. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 0ce5539 Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 726bfa5 Aug 29, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8 branch from f24dbe0 to 4536b50 Compare August 29, 2026 22:07
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 726bfa5 Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to d5834a8 Aug 30, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8 branch from 4536b50 to b234dae Compare August 30, 2026 07:25
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to d5834a8 Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to d3dd1c2 Aug 30, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8 branch from b234dae to a71bbe5 Compare August 30, 2026 14:27
@alebedev87

Copy link
Copy Markdown
Contributor

/ok-to-test

@openshift-ci openshift-ci Bot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Aug 30, 2026
…77d0

Image created from 'https://github.com/openshift/external-dns-operator?rev=0ce57338b52aae45afdf19e92096d0278601c7b7'

Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to d3dd1c2 Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 9d977d0 Aug 31, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-2-rhel-8-component-update-external-dns-operator-container-ext-dns-optr-1-2-rhel-8 branch from a71bbe5 to a94354b Compare August 31, 2026 00:09
@grzpiotrowski

grzpiotrowski commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

/test e2e-aws-shared-vpc-phz-operator
/retitle NE-2819: Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 9d977d0

@openshift-ci openshift-ci Bot changed the title Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 9d977d0 NE-2819: Update external-dns-operator-container-ext-dns-optr-1-2-rhel-8 to 9d977d0 Aug 31, 2026
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 31, 2026
@openshift-ci-robot

openshift-ci-robot commented Aug 31, 2026

Copy link
Copy Markdown

@red-hat-konflux-kflux-prd-rh03[bot]: This pull request references NE-2819 which is a valid jira issue.

Details

In response to this:

Image created from 'https://github.com/openshift/external-dns-operator?rev=0ce57338b52aae45afdf19e92096d0278601c7b7'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/external-dns-operator-tenant/ext-dns-optr-1-2-rhel-8/external-dns-operator-container-ext-dns-optr-1-2-rhel-8 digest 7c60e6a -> 9d977d0

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

@red-hat-konflux-kflux-prd-rh03[bot]: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-shared-vpc-phz-operator a94354b link true /test e2e-aws-shared-vpc-phz-operator

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. konflux-nudge ok-to-test Indicates a non-member PR verified by an org member that is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants