Skip to content

Pin AIDE version to 0.16 - #1041

Merged
Vincent056 merged 1 commit into
openshift:masterfrom
yuumasato:pin-aide-0.16
Aug 31, 2026
Merged

Pin AIDE version to 0.16#1041
Vincent056 merged 1 commit into
openshift:masterfrom
yuumasato:pin-aide-0.16

Conversation

@yuumasato

Copy link
Copy Markdown
Contributor

The config we are using in FIO is compatible with AIDE 0.16.

Migrating to 0.18 or 0.19 will require migration steps.

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 5, 2026
Pin the RPM lock file as well.
@openshift-ci

openshift-ci Bot commented Aug 5, 2026

Copy link
Copy Markdown
Contributor

@yuumasato: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@yuumasato

Copy link
Copy Markdown
Contributor Author

Follow up from #1042

@Vincent056 Vincent056 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@Vincent056
Vincent056 merged commit 35fa59a into openshift:master Aug 31, 2026
8 of 9 checks passed
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 31, 2026
@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Vincent056, yuumasato

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [Vincent056,yuumasato]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants