Skip to content

Update CODEOWNERS to change ownership to @shopiakkh - #2

Merged
shopiakkh merged 1 commit into
developfrom
shopiakkh-patch-2
Sep 4, 2026
Merged

Update CODEOWNERS to change ownership to @shopiakkh#2
shopiakkh merged 1 commit into
developfrom
shopiakkh-patch-2

Conversation

@shopiakkh

Copy link
Copy Markdown
Owner

No description provided.

@shopiakkh
shopiakkh merged commit 313685c into develop Sep 4, 2026
1 check failed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant