Skip to content

Add CODEOWNERS file for repository ownership - #5

Merged
shopiakkh merged 1 commit into
mainfrom
shopiakkh-patch-5
Sep 4, 2026
Merged

Add CODEOWNERS file for repository ownership#5
shopiakkh merged 1 commit into
mainfrom
shopiakkh-patch-5

Conversation

@shopiakkh

Copy link
Copy Markdown
Owner

No description provided.

@shopiakkh
shopiakkh merged commit 22db744 into main Sep 4, 2026
1 check failed
@shopiakkh
shopiakkh deleted the shopiakkh-patch-5 branch September 4, 2026 13:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant